waybar: refactor - #250099
Merged
Merged
waybar: refactor#250099
Conversation
- Reorder argument list - Use rec-less, overlay-style overridable recursive attributes (in effect since #119942); - Use pushd/popd instead of subshell - Use lib.mesonEnable and lib.mesonBool - Indentation - Remove nested `with`
ofborg
Bot
requested review from
FlorianFranzen,
Synthetica9,
jtbx,
lovesegfault,
minijackson and
rodrgz
August 19, 2023 04:58
AndersonTorres
marked this pull request as ready for review
August 19, 2023 14:30
Member
|
Now, I'm a bit confused about this PR. No checkbox was ticked, so I don't know the extent of the tests you did, and it was merged only a few hours later, without any other comment by someone else. I was planning on reviewing it when I got notified, but simply did not have the time to finish work and go home before the PR was merged. I would probably have accepted, I do like the changes, but you can guess my surprise when I found out that a package that I maintain got refactored without any approval of any of the maintainers. |
Member
Author
|
I judged the reword as mostly equivalent and needed no more than a local testing and an all-green CI. Sorry by my haste. I was on a small hurry for merging some already approved PRs. |
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
withDescription of changes
Things done
sandbox = trueset innix.conf? (See Nix manual)nix-shell -p nixpkgs-review --run "nixpkgs-review rev HEAD". Note: all changes have to be committed, also see nixpkgs-review usage./result/bin/)