Repository navigation
highs: init at 1.6.0 - #273289
highs: init at 1.6.0#273289david-r-cox wants to merge 2 commits into
Conversation
|
You need to add yourself to |
Thanks @quantenzitrone -- I updated |
eclairevoyant
left a comment
There was a problem hiding this comment.
As far as the maintainer commit, can you rebase this branch to come off 6290195 where you added yourself to maintainers? That way you're not really duplicating work, but ofborg can eval it.
|
Result of 1 package built:
|
eclairevoyant
left a comment
There was a problem hiding this comment.
One more bit of feedback, but otherwise tested out on my end, seems good
|
@eclairevoyant is there anything else that I should be doing on this PR? This (along with #273296, #273295, and #273252) is my first time contributing to nixpkgs. Thanks again for the review. |
|
I approved but I don't have access to merge |
| homepage = "https://www.highs.dev"; | ||
| license = with licenses; [ mit ]; | ||
| maintainers = with maintainers; [ david-r-cox ]; | ||
| mainProgram = "highs"; |
There was a problem hiding this comment.
Looking at the CI again - do you know why the tests are failing on darwin?
|
looks like #295386 already got merged |
Description of changes
Things done
nix.conf? (See Nix manual)sandbox = relaxedsandbox = truenix-shell -p nixpkgs-review --run "nixpkgs-review rev HEAD". Note: all changes have to be committed, also see nixpkgs-review usage./result/bin/)Add a 👍 reaction to pull requests you find important.