Skip to content

kotlin-lsp: init at 261.13587.0 - #492163

Closed
Badya wants to merge 1 commit into
NixOS:masterfrom
Badya:master
Closed

Badya wants to merge 1 commit into
NixOS:masterfrom
Badya:master

Conversation

@Badya

@Badya Badya commented Feb 19, 2026 •

Copy link
Copy Markdown

Official JetBrains implementation of LSP for Kotlin language
https://github.com/Kotlin/kotlin-lsp

Things done

  • Built on platform:
    • x86_64-linux
    • aarch64-linux
    • x86_64-darwin
    • aarch64-darwin
  • Tested, as applicable:
  • Ran nixpkgs-review on this PR. See nixpkgs-review usage.
  • Tested basic functionality of all binary files, usually in ./result/bin/.
  • Nixpkgs Release Notes
    • Package update: when the change is major or breaking.
  • NixOS Release Notes
    • Module addition: when adding a new NixOS module.
    • Module update: when the change is significant.
  • Fits CONTRIBUTING.md, pkgs/README.md, maintainers/README.md and other READMEs.

@nixpkgs-ci nixpkgs-ci Bot added 8.has: package (new) This PR adds a new package 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux. 10.rebuild-darwin: 1-10 This PR causes between 1 and 10 packages to rebuild on Darwin. 10.rebuild-darwin: 1 This PR causes 1 package to rebuild on Darwin. 10.rebuild-linux: 1 This PR causes 1 package to rebuild on Linux. 12.first-time contribution This PR is the author's first one; please be gentle! 9.needs: reviewer This PR currently has no reviewers requested and needs attention. labels Feb 19, 2026
@Badya
Badya force-pushed the master branch 2 times, most recently from 67d34a9 to 99ece41 Compare February 20, 2026 14:28
@theCapypara

theCapypara commented Feb 23, 2026 •

Copy link
Copy Markdown
Member

This is a duplicate of #482845 and #435169

@bew bew mentioned this pull request Apr 29, 2026
4 of 13 tasks
@Badya

Badya commented Apr 29, 2026

Copy link
Copy Markdown
Author

Closed in favour of #514623

@Badya Badya closed this Apr 29, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

8.has: package (new) This PR adds a new package 9.needs: reviewer This PR currently has no reviewers requested and needs attention. 10.rebuild-darwin: 1-10 This PR causes between 1 and 10 packages to rebuild on Darwin. 10.rebuild-darwin: 1 This PR causes 1 package to rebuild on Darwin. 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux. 10.rebuild-linux: 1 This PR causes 1 package to rebuild on Linux. 12.first-time contribution This PR is the author's first one; please be gentle!

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants