kodi: replace source urls for apache dependencies with stable ones - #304845
Merged
SuperSandro2000 merged 1 commit intoApr 18, 2024
Merged
Conversation
6 of 13 tasks
nvmd
force-pushed
the
master+kodi-fix-apache-commons
branch
from
April 18, 2024 06:55
f5ed9b6 to
d88c3ac
Compare
nvmd
force-pushed
the
master+kodi-fix-apache-commons
branch
from
April 18, 2024 06:57
d88c3ac to
67d4f93
Compare
nvmd
requested review from
RaghavSood,
RaitoBezarius,
jonringer and
mmahut
as code owners
April 18, 2024 06:57
Co-authored-by: Sandro <sandro.jaeckel@gmail.com>
nvmd
force-pushed
the
master+kodi-fix-apache-commons
branch
from
April 18, 2024 06:59
67d4f93 to
861d189
Compare
SuperSandro2000
approved these changes
Apr 18, 2024
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Description of changes
Replaced source of apache
groovy,commons-lang3, andcommons-textwith stable ones, they contain only the latest version:commons-textis already not there:https://dlcdn.apache.org/commons/text/binaries/commons-text-1.11.0-bin.zipreturns 404, breaking the build of Kodi 20.1 just merged in #303497.See in a build attempt of #304097: https://logs.ofborg.org/?key=nixos/nixpkgs.304097&attempt_id=45c6a77c-3113-46e8-a971-e5bd1f514559
Things done
nix.conf? (See Nix manual)sandbox = relaxedsandbox = truenix-shell -p nixpkgs-review --run "nixpkgs-review rev HEAD". Note: all changes have to be committed, also see nixpkgs-review usage./result/bin/)Add a 👍 reaction to pull requests you find important.